Skip to content

fix: upgrade electron-builder keychain signing - #392

Merged
dory-finn merged 1 commit into
mainfrom
nextjsv163
Sep 8, 2026
Merged

dory-finn merged 1 commit into
mainfrom
nextjsv163

Conversation

@dory-finn

Copy link
Copy Markdown
Contributor

No description provided.

@vercel

vercel Bot commented Sep 8, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

3 Skipped Deployments
Project Deployment Actions Updated
dory Ignored Ignored Sep 8, 2026 7:45am UTC
dory-admin Ignored Ignored Sep 8, 2026 7:45am UTC
dory-beta Ignored Ignored Sep 8, 2026 7:45am UTC

Request Review

@dory-finn
dory-finn merged commit 2928eff into main Sep 8, 2026
8 of 9 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant