Skip to content

update: maintainers list and evolution resources - #534

Open
poiana wants to merge 1 commit into
mainfrom
update-evolution-files
Open

poiana wants to merge 1 commit into
mainfrom
update-evolution-files

Conversation

@poiana

@poiana poiana commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

Updating maintainers list and the evolution resource files. Made using the update-maintainers periodic ProwJob. Do not edit this PR.

In case you wanna change your name or your company change this file.

/kind documentation

@poiana poiana added kind/documentation Improvements or additions to documentation dco-signoff: yes labels Sep 15, 2026
@poiana

poiana commented Sep 15, 2026

Copy link
Copy Markdown
Contributor Author

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: poiana
Once this PR has been reviewed and has the lgtm label, please assign lucaguerra for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana poiana added the size/XS label Sep 15, 2026
@poiana
poiana force-pushed the update-evolution-files branch 2 times, most recently from b8b92c7 to aa60b58 Compare September 17, 2026 09:04
Signed-off-by: poiana <51138685+poiana@users.noreply.github.com>
@poiana
poiana force-pushed the update-evolution-files branch from aa60b58 to 8b70317 Compare September 18, 2026 09:04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes kind/documentation Improvements or additions to documentation size/XS

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant