Skip to content

new(ci): check PRs for semver violations - #76

Merged
poiana merged 1 commit into
falcosecurity:mainfrom
gnosek:semver-checks
Dec 10, 2025
Merged

poiana merged 1 commit into
falcosecurity:mainfrom
gnosek:semver-checks

Conversation

@gnosek

@gnosek gnosek commented Nov 21, 2025

Copy link
Copy Markdown
Collaborator

What type of PR is this?

Uncomment one (or more) /kind <> lines:

/kind bug

/kind cleanup

/kind design

/kind documentation

/kind failing-test

/kind feature

Any specific area of the project related to this PR?

Uncomment one (or more) /area <> lines:

/area build

/area ci

/area event

/area event_derive

/area plugin

/area plugin_api

/area plugin_derive

/area plugin_tests

What this PR does / why we need it:

(Try to) flag PRs which break the public API without an explicit ! marker in the title

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

Does this PR introduce a user-facing change?:

NONE

@poiana
poiana requested review from ekoops and mrgian November 21, 2025 21:13
@gnosek gnosek changed the title wip: new(ci): check PRs for semver violations wip: new(ci)!: check PRs for semver violations Nov 21, 2025
@gnosek gnosek changed the title wip: new(ci)!: check PRs for semver violations wip: new(ci): check PRs for semver violations Nov 21, 2025
Signed-off-by: Grzegorz Nosek <grzegorz.nosek@sysdig.com>
@gnosek gnosek changed the title wip: new(ci): check PRs for semver violations new(ci): check PRs for semver violations Nov 21, 2025
@ekoops

ekoops commented Nov 22, 2025

Copy link
Copy Markdown
Contributor

I'm wondering if it is possible to catch for PRs providing the ! marker in the title and indeed not breaking anything...

@gnosek

gnosek commented Nov 22, 2025

Copy link
Copy Markdown
Collaborator Author

I'm wondering if it is possible to catch for PRs providing the ! marker in the title and indeed not breaking anything...

Sure, just need to tweak the action a bit, though it... is... pretty... slow... (slower than a build+test run on linux) and I'm not sure it is all that beneficial, but I'm happy to be convinced (in any case, it's much faster than a full benchmark run)

@ekoops

ekoops commented Dec 10, 2025

Copy link
Copy Markdown
Contributor

Yeah, not an issue. We can take it as is and introduce the check later if need arises

@ekoops ekoops left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

@poiana

poiana commented Dec 10, 2025

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 0bf3cebd1294d2640fbd340b39c445a23b501980

@poiana

poiana commented Dec 10, 2025

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ekoops, gnosek

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana
poiana merged commit f7c1ab1 into falcosecurity:main Dec 10, 2025
8 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants