Skip to content

chore(deps): Bump keccak from 0.1.5 to 0.1.6 - #78

Merged
poiana merged 1 commit into
mainfrom
dependabot/cargo/keccak-0.1.6
Mar 2, 2026
Merged

poiana merged 1 commit into
mainfrom
dependabot/cargo/keccak-0.1.6

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Feb 20, 2026

Copy link
Copy Markdown
Contributor

Bumps keccak from 0.1.5 to 0.1.6.

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)
    You can disable automated security fix PRs for this repo from the Security Alerts page.

Bumps [keccak](https://github.com/RustCrypto/sponges) from 0.1.5 to 0.1.6.
- [Commits](RustCrypto/sponges@keccak-v0.1.5...keccak-v0.1.6)

---
updated-dependencies:
- dependency-name: keccak
  dependency-version: 0.1.6
  dependency-type: indirect
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file rust Pull requests that update rust code labels Feb 20, 2026
@poiana

poiana commented Feb 20, 2026

Copy link
Copy Markdown

There is not a label identifying the kind of this PR.
Please specify it either using /kind <group> or manually from the side menu.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@ekoops ekoops left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

@poiana

poiana commented Mar 2, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dependabot[bot], ekoops

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana

poiana commented Mar 2, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: caab2a6f398f0a5d721a4a8b3b3823bea597528b

@poiana poiana added the approved label Mar 2, 2026
@ekoops

ekoops commented Mar 2, 2026

Copy link
Copy Markdown
Contributor

/release-note-none

@poiana
poiana merged commit a7fb2f1 into main Mar 2, 2026
7 checks passed
@poiana
poiana deleted the dependabot/cargo/keccak-0.1.6 branch March 2, 2026 09:07
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved dco-signoff: yes dependencies Pull requests that update a dependency file lgtm needs-kind release-note-none rust Pull requests that update rust code size/XS

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants