Skip to content

fix(.github/workflow): correct variable injection in semver-checks - #79

Merged
poiana merged 1 commit into
mainfrom
fix/github/workflow
Mar 2, 2026
Merged

poiana merged 1 commit into
mainfrom
fix/github/workflow

Conversation

@leogr

@leogr leogr commented Mar 2, 2026

Copy link
Copy Markdown
Member

What type of PR is this?

Uncomment one (or more) /kind <> lines:

/kind bug

/kind cleanup

/kind design

/kind documentation

/kind failing-test

/kind feature

Any specific area of the project related to this PR?

Uncomment one (or more) /area <> lines:

/area build

/area ci

/area event

/area event_derive

/area plugin

/area plugin_api

/area plugin_derive

/area plugin_tests

What this PR does / why we need it:

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

Does this PR introduce a user-facing change?:

NONE

Signed-off-by: Leonardo Grasso <me@leonardograsso.com>
@poiana

poiana commented Mar 2, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: gnosek, leogr

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana poiana added the approved label Mar 2, 2026
@poiana
poiana merged commit 14d385c into main Mar 2, 2026
7 checks passed
@poiana
poiana deleted the fix/github/workflow branch March 2, 2026 14:56
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants