Skip to content

ci(.github): replace /area CI with /area automation in PR template - #80

Merged
poiana merged 1 commit into
falcosecurity:mainfrom
ekoops:ekoops/area-automation
May 19, 2026
Merged

poiana merged 1 commit into
falcosecurity:mainfrom
ekoops:ekoops/area-automation

Conversation

@ekoops

@ekoops ekoops commented May 7, 2026

Copy link
Copy Markdown
Contributor

What type of PR is this?

Uncomment one (or more) /kind <> lines:

/kind bug

/kind cleanup

/kind design

/kind documentation

/kind failing-test

/kind feature

Any specific area of the project related to this PR?

Uncomment one (or more) /area <> lines:

/area build

/area automation

/area event

/area event_derive

/area plugin

/area plugin_api

/area plugin_derive

/area plugin_tests

What this PR does / why we need it:

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

Does this PR introduce a user-facing change?:

NONE

Signed-off-by: Leonardo Di Giovanna <leonardodigiovanna1@gmail.com>
@ekoops
ekoops force-pushed the ekoops/area-automation branch from 2a286db to 9b4f5eb Compare May 7, 2026 15:03
@poiana

poiana commented May 19, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ekoops, gnosek

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana
poiana merged commit 7b205ac into falcosecurity:main May 19, 2026
7 checks passed
@ekoops
ekoops deleted the ekoops/area-automation branch May 19, 2026 14:20
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants