Skip to content

chore(deps): upgrade go dependencies - #334

Closed
jenkins-x-bot wants to merge 1 commit into
mainfrom
pr-e3f1805e-f086-42c0-8be7-1c31943dfb07
Closed

jenkins-x-bot wants to merge 1 commit into
mainfrom
pr-e3f1805e-f086-42c0-8be7-1c31943dfb07

Conversation

@jenkins-x-bot

Copy link
Copy Markdown
Contributor

@jenkins-x-bot

Copy link
Copy Markdown
Contributor Author

@jenkins-x-bot: The following test failed, say /retest to rerun them all:

Test name Commit Details Rerun command
pr 5c210d1 link /test
Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the jenkins-x/lighthouse repository. I understand the commands that are listed here.

@msvticket msvticket closed this Aug 2, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants