Skip to content

Use the Ginkgo-provided context in tests - #76

Closed
skitt wants to merge 1 commit into
kubernetes-sigs:masterfrom
skitt:ginkgo-context
Closed

Use the Ginkgo-provided context in tests#76
skitt wants to merge 1 commit into
kubernetes-sigs:masterfrom
skitt:ginkgo-context

Conversation

@skitt

@skitt skitt commented Sep 4, 2026

Copy link
Copy Markdown
Member

This allows the tests to be canceled by Ginkgo.

This allows the tests to be canceled by Ginkgo.

Signed-off-by: Stephen Kitt <skitt@redhat.com>
@kubernetes-prow kubernetes-prow Bot added the sig/multicluster Categorizes an issue or PR as relevant to SIG Multicluster. label Sep 4, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: skitt

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Sep 4, 2026
@skitt skitt closed this Sep 4, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. sig/multicluster Categorizes an issue or PR as relevant to SIG Multicluster. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant