Skip to content

[stable33] chore(deps-dev): bump sass from 1.104.0 to 1.105.0 - #64546

Merged
nextcloud-command merged 2 commits into
stable33from
dependabot/npm_and_yarn/stable33/sass-1.104.1
Sep 26, 2026
Merged

nextcloud-command merged 2 commits into
stable33from
dependabot/npm_and_yarn/stable33/sass-1.104.1

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 19, 2026 •

Copy link
Copy Markdown
Contributor

Bumps sass from 1.104.0 to 1.105.0.

Release notes

Sourced from sass's releases.

Dart Sass 1.105.0

To install Sass 1.105.0, download one of the packages below and add it to your PATH, or see the Sass website for full installation instructions.

Changes

  • Add support for first-class modules. These can be accessed using the new meta.load() and meta.get-module() functions, and may be passed as the $module argument to numerous eisting sass:meta functions.

  • Add the meta.css() mixin, which includes CSS from a first-class module.

JS API

  • Add a SassModule class and a corresponding Value.assertModule() method.

Dart API

  • Add a SassModule class and a corresponding Value.assertModule() method.

See the full changelog for changes in earlier releases.

Dart Sass 1.104.1

To install Sass 1.104.1, download one of the packages below and add it to your PATH, or see the Sass website for full installation instructions.

Changes

  • Fix a bug where loud comments before @use rules could be emitted multiple times under certain circumstances.

  • Fix a bug in which an invalid selector at the end of an indented syntax stylesheet would cause the parser to crash instead of emitting a useful error message.

Command Line Interface

  • Many-to-many compilations no longer compile any files that appear in the output directory if the output directory is also within the source directory. This fixes a bug where --watch mode could enter an infinite loop recompiling the same CSS file over and over.

  • Sass now sets the modification time of output files to the time compilation started rather than the time it ended. This ensures that, if a source file is modified during compilation, --watch and --update mode will recompile the outputs to include the new source file contents.

See the full changelog for changes in earlier releases.

Changelog

Sourced from sass's changelog.

1.105.0

  • Add support for first-class modules. These can be accessed using the new meta.load() and meta.get-module() functions, and may be passed as the $module argument to numerous eisting sass:meta functions.

  • Add the meta.css() mixin, which includes CSS from a first-class module.

JS API

  • Add a SassModule class and a corresponding Value.assertModule() method.

Dart API

  • Add a SassModule class and a corresponding Value.assertModule() method.

1.104.1

  • Fix a bug where loud comments before @use rules could be emitted multiple times under certain circumstances.

  • Fix a bug in which an invalid selector at the end of an indented syntax stylesheet would cause the parser to crash instead of emitting a useful error message.

Command Line Interface

  • Many-to-many compilations no longer compile any files that appear in the output directory if the output directory is also within the source directory. This fixes a bug where --watch mode could enter an infinite loop recompiling the same CSS file over and over.

  • Sass now sets the modification time of output files to the time compilation started rather than the time it ended. This ensures that, if a source file is modified during compilation, --watch and --update mode will recompile the outputs to include the new source file contents.

Commits
  • 4bf2b92 Bump zizmorcore/zizmor-action from 0.6.3 to 0.6.4 (#2865)
  • 850d57e [meta.load] Add support for meta.load() and related features (#2861)
  • 6180be0 Bump postcss from 8.5.26 to 8.5.28 in /pkg/sass-parser (#2863)
  • da84fb5 Fix --watch not recompiling sources modified during a compilation (#2850)
  • 8a287a7 Merge pull request #2859 from sass/fix-auto-merge
  • 6033b46 Add explicit versions to .github/util actions
  • cc71d43 Only auto-merge dependabot PRs once the tests pass
  • 49ddc4f Revert "Bump dart-lang/setup-dart in /.github/util/initialize"
  • 71a127d Avoid duplicating loud comments that appear before @use (#2854)
  • f19baab Merge pull request #2857 from sass/dependabot/github_actions/dot-github/util/...
  • Additional commits viewable in compare view

@dependabot
dependabot Bot requested review from a team as code owners September 19, 2026 03:07
@dependabot
dependabot Bot requested review from kristian-zendato, sorbaugh and susnux and removed request for a team September 19, 2026 03:07
@github-actions github-actions Bot changed the title chore(deps-dev): bump sass from 1.104.0 to 1.104.1 [stable33] chore(deps-dev): bump sass from 1.104.0 to 1.104.1 Sep 19, 2026
@AndyScherzinger AndyScherzinger added this to the Nextcloud 33.0.10 milestone Sep 22, 2026
@AndyScherzinger

Copy link
Copy Markdown
Member

@dependabot rebase

@dependabot dependabot Bot changed the title [stable33] chore(deps-dev): bump sass from 1.104.0 to 1.104.1 chore(deps-dev): bump sass from 1.104.0 to 1.105.0 Sep 26, 2026
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/stable33/sass-1.104.1 branch from 57a91df to ab6c978 Compare September 26, 2026 18:49
@github-actions github-actions Bot changed the title chore(deps-dev): bump sass from 1.104.0 to 1.105.0 [stable33] chore(deps-dev): bump sass from 1.104.0 to 1.105.0 Sep 26, 2026
Bumps [sass](https://github.com/sass/dart-sass) from 1.104.0 to 1.105.0.
- [Release notes](https://github.com/sass/dart-sass/releases)
- [Changelog](https://github.com/sass/dart-sass/blob/main/CHANGELOG.md)
- [Commits](sass/dart-sass@1.104.0...1.105.0)

---
updated-dependencies:
- dependency-name: sass
  dependency-version: 1.104.1
  dependency-type: direct:development
  update-type: version-update:semver-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@AndyScherzinger
AndyScherzinger force-pushed the dependabot/npm_and_yarn/stable33/sass-1.104.1 branch from ab6c978 to 7ecc981 Compare September 26, 2026 19:12
@AndyScherzinger

Copy link
Copy Markdown
Member

/compile

Signed-off-by: nextcloud-command <nextcloud-command@users.noreply.github.com>
@nextcloud-command
nextcloud-command merged commit d700399 into stable33 Sep 26, 2026
123 of 124 checks passed
@nextcloud-command
nextcloud-command deleted the dependabot/npm_and_yarn/stable33/sass-1.104.1 branch September 26, 2026 19:37
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants