Skip to content
This repository was archived by the owner on Aug 11, 2026. It is now read-only.

Use nvidia runtime handler for the daemonset - #966

Merged
openshift-merge-bot[bot] merged 3 commits into
openshift:nextfrom
harche:runtime_fix
Dec 1, 2025
Merged

openshift-merge-bot[bot] merged 3 commits into
openshift:nextfrom
harche:runtime_fix

Conversation

@harche

@harche harche commented Nov 21, 2025 •

Copy link
Copy Markdown
Contributor

NVIDIA/gpu-operator#1578 PR broke DAS operator daemonset, so now we need to set runtimeClassName: nvidia going forward for DAS operator daemonset since all access to the gpu is handled by using CDI which handles mounting the required nvml libraries.

For testing, If I set NVIDIA_RUNTIME_SET_AS_DEFAULT=true in driver took in gpu cluster policy, DAS daemonset pod starts working again,

% oc get pods -n das-operator
NAME                                    READY   STATUS    RESTARTS       AGE
das-daemonset-2f9r4                     1/1     Running   10 (11m ago)   32m
das-operator-67598957b7-xqng7           1/1     Running   0              33m
das-operator-67598957b7-zqbt2           1/1     Running   0              33m
das-operator-webhook-55c7778587-5mpj8   1/1     Running   0              32m
das-operator-webhook-55c7778587-tbzls   1/1     Running   0              32m
das-scheduler-67dccd6f9f-bd8xd          1/1     Running   0              32m
das-scheduler-67dccd6f9f-sqvpl          1/1     Running   0              32m

Fixes : https://issues.redhat.com/browse/OCPBUGS-65805

@openshift-ci openshift-ci Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Nov 21, 2025
@openshift-ci
openshift-ci Bot requested review from asm582 and cpmeadors November 21, 2025 14:58
@openshift-ci

openshift-ci Bot commented Nov 21, 2025

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: harche

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Nov 21, 2025
@harche

harche commented Nov 23, 2025

Copy link
Copy Markdown
Contributor Author

/test e2e-bundle-4-19-runc

@harche

harche commented Nov 23, 2025

Copy link
Copy Markdown
Contributor Author

/test e2e-bundle-runc

Signed-off-by: Harshal Patil <12152047+harche@users.noreply.github.com>
Signed-off-by: Harshal Patil <12152047+harche@users.noreply.github.com>
@harche

harche commented Nov 25, 2025

Copy link
Copy Markdown
Contributor Author

/test e2e-bundle-runc

@harche

harche commented Nov 25, 2025

Copy link
Copy Markdown
Contributor Author

/test e2e-bundle-4-19-runc

@openshift-ci openshift-ci Bot added size/S Denotes a PR that changes 10-29 lines, ignoring generated files. and removed size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Nov 26, 2025
Signed-off-by: Harshal Patil <12152047+harche@users.noreply.github.com>
@harche

harche commented Nov 26, 2025

Copy link
Copy Markdown
Contributor Author

/test e2e-bundle-runc

@openshift-ci

openshift-ci Bot commented Nov 26, 2025

Copy link
Copy Markdown

@harche: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@rphillips

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Dec 1, 2025
@openshift-merge-bot
openshift-merge-bot Bot merged commit 2c8fc17 into openshift:next Dec 1, 2025
6 checks passed
@harche
harche deleted the runtime_fix branch December 1, 2025 16:12
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants