Skip to content

fix(deps): bump actions/setup-python from v2.1.4 to v2.2.2 - #97

Closed
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/github_actions/actions/setup-python-v2.2.2
Closed

fix(deps): bump actions/setup-python from v2.1.4 to v2.2.2#97
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/github_actions/actions/setup-python-v2.2.2

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Apr 13, 2021

Copy link
Copy Markdown
Contributor

Bumps actions/setup-python from v2.1.4 to v2.2.2.

Release notes

Sourced from actions/setup-python's releases.

v2.2.2

Address PyPy installation issues on Windows: actions/setup-python#196

v2.2.1

  • Fix PyPy3 issues with Windows

v2.2.0

  • Support for PyPy 3.7 in addition to the current 3.6 support
  • Support for Pre-Release distributions of PyPy
  • Support for PyPy Nighly
Commits
  • dc73133 Fix PyPy installation on Windows to adopt new parameters format (#201)
  • a112144 Add on: pull_request trigger to CodeQL workflow (#180)
  • 66319ca Use quotes around Python versions in README (#175)
  • 3105fb1 fix is_windows (#172)
  • 8c5ea63 Adding support for more PyPy versions and installing them on-flight (#168)
  • 2831efe Improve find-python to add "Scripts" folder to PATH on Windows machines (#169)
  • 3b3f2de update pypy3 to point to 3.6 (#164)
  • 723e46d CODEOWNERS needs the org name for teams
  • 195f5c3 Create CODEOWNERS
  • See full diff in compare view

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot merge will merge this PR after your CI passes on it
  • @dependabot squash and merge will squash and merge this PR after your CI passes on it
  • @dependabot cancel merge will cancel a previously requested merge and block automerging
  • @dependabot reopen will reopen this PR if it is closed
  • @dependabot close will close this PR and stop Dependabot recreating it. You can achieve the same result by closing it manually
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [actions/setup-python](https://github.com/actions/setup-python) from v2.1.4 to v2.2.2.
- [Release notes](https://github.com/actions/setup-python/releases)
- [Commits](actions/setup-python@v2.1.4...dc73133)

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot @github

dependabot Bot commented on behalf of github Apr 13, 2021

Copy link
Copy Markdown
Contributor Author

Dependabot tried to add @sundowndev as a reviewer to this PR, but received the following error from GitHub:

POST https://api.github.com/repos/tsiw247/PhoneInfoga/pulls/97/requested_reviewers: 422 - Reviews may only be requested from collaborators. One or more of the users or teams you specified is not a collaborator of the tsiw247/PhoneInfoga repository. // See: https://docs.github.com/rest/reference/pulls#request-reviewers-for-a-pull-request

@dependabot dependabot Bot added dependencies Pull requests that update a dependency file github_actions Pull requests that update Github_actions code labels Apr 13, 2021
@codecov-io

Copy link
Copy Markdown

Codecov Report

❗ No coverage uploaded for pull request base (master@6c9ce54). Click here to learn what that means.
The diff coverage is n/a.

Impacted file tree graph

@@            Coverage Diff            @@
##             master      #97   +/-   ##
=========================================
  Coverage          ?   94.32%           
=========================================
  Files             ?       11           
  Lines             ?      529           
  Branches          ?        0           
=========================================
  Hits              ?      499           
  Misses            ?       22           
  Partials          ?        8           

Continue to review full report at Codecov.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update 6c9ce54...55f4398. Read the comment docs.

@dependabot @github

dependabot Bot commented on behalf of github May 12, 2021

Copy link
Copy Markdown
Contributor Author

Superseded by #118.

@dependabot dependabot Bot closed this May 12, 2021
@dependabot
dependabot Bot deleted the dependabot/github_actions/actions/setup-python-v2.2.2 branch May 12, 2021 00:05
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file github_actions Pull requests that update Github_actions code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant